Skip to content

chore(deps): update quay.io/rhacs-eng/konflux-tasks:latest docker digest to f1d2954 - #3986

Merged
red-hat-konflux[bot] merged 1 commit into
release-2.39from
konflux/mintmaker/release-2.39/stackrox-custom-konflux-tasks
Oct 10, 2026
Merged

red-hat-konflux[bot] merged 1 commit into
release-2.39from
konflux/mintmaker/release-2.39/stackrox-custom-konflux-tasks

Conversation

@red-hat-konflux

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
quay.io/rhacs-eng/konflux-tasks tekton-bundle digest 81ed7b7 → f1d2954

Configuration

📅 Schedule: (in timezone Etc/UTC)

  • Branch creation
    • Between 12:00 AM and 07:59 AM (* 0-7 * * *)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

…est to f1d2954

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot requested review from a team and rhacs-bot as code owners October 10, 2026 03:46
@red-hat-konflux
red-hat-konflux Bot enabled auto-merge (squash) October 10, 2026 03:46
@red-hat-konflux
red-hat-konflux Bot requested a review from a team as a code owner October 10, 2026 03:46

@rhacs-bot rhacs-bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Auto-approved by automation.

@openshift-ci

openshift-ci Bot commented Oct 10, 2026

Copy link
Copy Markdown

@red-hat-konflux[bot]: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/slim-e2e-tests 00dbf5d link false /test slim-e2e-tests

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@red-hat-konflux
red-hat-konflux Bot merged commit 33cb24a into release-2.39 Oct 10, 2026
40 of 44 checks passed
@red-hat-konflux
red-hat-konflux Bot deleted the konflux/mintmaker/release-2.39/stackrox-custom-konflux-tasks branch October 10, 2026 04:30
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant