Skip to content

chore(deps): update all updates - #3963

Merged
red-hat-konflux[bot] merged 1 commit into
release-2.40from
konflux/mintmaker/release-2.40/all-updates
Oct 8, 2026
Merged

red-hat-konflux[bot] merged 1 commit into
release-2.40from
konflux/mintmaker/release-2.40/all-updates

Conversation

@red-hat-konflux

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
registry.access.redhat.com/ubi9-minimal final digest 1d7c551 → 5ed244b
registry.redhat.io/rhel9/postgresql-15 final digest 65c634b → 689960f

Configuration

📅 Schedule: (in timezone Etc/UTC)

  • Branch creation
    • Between 12:00 AM and 07:59 AM (* 0-7 * * *)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@red-hat-konflux
red-hat-konflux Bot requested review from a team and rhacs-bot as code owners October 7, 2026 01:43
@red-hat-konflux
red-hat-konflux Bot enabled auto-merge (squash) October 7, 2026 01:44

@rhacs-bot rhacs-bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Auto-approved by automation.

@rhacs-bot rhacs-bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Auto-approved by automation.

@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/release-2.40/all-updates branch from ab69537 to 6ab5f12 Compare October 7, 2026 05:29
@github-actions

github-actions Bot commented Oct 7, 2026

Copy link
Copy Markdown

/konflux-retest scanner-db-slim-on-push

1 similar comment
@github-actions

github-actions Bot commented Oct 7, 2026

Copy link
Copy Markdown

/konflux-retest scanner-db-slim-on-push

@github-actions

github-actions Bot commented Oct 7, 2026

Copy link
Copy Markdown

/konflux-retest scanner-on-push

@github-actions

github-actions Bot commented Oct 7, 2026

Copy link
Copy Markdown

/konflux-retest scanner-db-slim-on-push

@github-actions

github-actions Bot commented Oct 7, 2026

Copy link
Copy Markdown

/konflux-retest scanner-on-push

1 similar comment
@github-actions

github-actions Bot commented Oct 7, 2026

Copy link
Copy Markdown

/konflux-retest scanner-on-push

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/release-2.40/all-updates branch from 6ab5f12 to 5434734 Compare October 8, 2026 04:40
@openshift-ci

openshift-ci Bot commented Oct 8, 2026

Copy link
Copy Markdown

@red-hat-konflux[bot]: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/slim-e2e-tests 6ab5f12 link false /test slim-e2e-tests
ci/prow/e2e-tests 5434734 link false /test e2e-tests

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@red-hat-konflux
red-hat-konflux Bot merged commit d9b2a28 into release-2.40 Oct 8, 2026
36 of 38 checks passed
@red-hat-konflux
red-hat-konflux Bot deleted the konflux/mintmaker/release-2.40/all-updates branch October 8, 2026 04:59
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant