Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: mumesan The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important Review skippedIgnore keyword(s) in the title. ⛔ Ignored keywords (1)
Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Central YAML (base), Organization UI (inherited) Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
OpenStackControlPlane CRD Size Report
Threshold reference
|
|
@mumesan: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 18m 45s |
|
@mumesan does it mean that inspector is gone and no longer used and there is no migration needed for any config which a user set for it before? |
Hi @stuggi , we've decided not to proceed with inspector removal from ironic-operator at this time and I'm going to close this PR. |
Description:
Inspector has been moved from a service to integrate with ironic as an agent, as a result all references to it need to be removed from ironic-operator. This PR removes it in openstack-operator so it compiles successfully when the references are removed from ironic-operator.