[main] Upgrade to latest dependencies - #4058
knative-prow[bot] merged 1 commit into
Conversation
|
@knative-automation: GitHub didn't allow me to request PR reviews from the following users: knative/func-approvers. Note that only knative members and repo collaborators can review this PR, and authors cannot review their own PRs. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
9fe3104 to
10979d8
Compare
|
/lgtm |
|
/retest |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: gauron99, knative-automation The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
bumping knative.dev/networking 17b28c6...e397784: > e397784 upgrade to latest dependencies (# 1153) > 7d916c7 Fix non-contant format string err in go 1.26 (# 1152) bumping knative.dev/client/pkg 5e89c92...63f8b66: > 63f8b66 upgrade to latest dependencies (# 2270) > 9978385 upgrade to latest dependencies (# 2269) bumping knative.dev/eventing 07b1e8a...7c892cf: > 7c892cf test(delivery): cover backoff maximum through triggers (# 9330) > 68ee8ee build(deps): bump google.golang.org/grpc from 1.83.1 to 1.83.2 (# 9334) > 66a5fa9 Fix RequestReply duplicate reply deadlock (# 9191) > 99df328 Fix non-contant format string err in go 1.26 (# 9406) bumping google.golang.org/grpc 1550d9e...030ee8b: > 030ee8b Update version to 1.83.2 (# 9375) > 8668b69 cherry-pick # 9365 to v1.83.x (# 9366) > a3e952d cherry-pick # 9346 to v1.83.x and update x/net dependency (# 9369) > 58f8fd9 Change version to 1.83.2-dev (# 9337) bumping knative.dev/pkg d2a153a...5dc1978: > 5dc1978 Bump google.golang.org/grpc from 1.83.0 to 1.83.2 (# 3389) bumping knative.dev/serving 618b0c2...47ddb67: > 47ddb67 Update net-contour nightly (# 16692) > 017bbdb upgrade to latest dependencies (# 16691) Signed-off-by: Knative Automation <automation@knative.team>
10979d8 to
160f714
Compare
|
New changes are detected. LGTM label has been removed. |
Cron -dprotaso
/cc knative/func-approvers
/assign knative/func-approvers
Produced by: knative-extensions/knobots/actions/update-deps