Repository navigation
Bump github.com/gardener/gardener from 1.142.1 to 1.142.6 - #415
dependabot[bot] wants to merge 1 commit into
Conversation
|
The PR needs to be labeled with ok-to-test by a maintainer to trigger the automated validation of the change |
|
LGTM label has been added. DetailsGit tree hash: afc0ae99e87efa5047f78138b752cbbf1d7c131d |
|
The PR needs to be labeled with ok-to-test by a maintainer to trigger the automated validation of the change |
Bumps [github.com/gardener/gardener](https://github.com/gardener/gardener) from 1.142.1 to 1.142.6. - [Release notes](https://github.com/gardener/gardener/releases) - [Commits](gardener/gardener@v1.142.1...v1.142.6) --- updated-dependencies: - dependency-name: github.com/gardener/gardener dependency-version: 1.142.6 dependency-type: direct:production ... Signed-off-by: dependabot[bot] <support@github.com>
54e318c to
c2cdf1a
Compare
|
The PR needs to be labeled with ok-to-test by a maintainer to trigger the automated validation of the change |
2 similar comments
|
The PR needs to be labeled with ok-to-test by a maintainer to trigger the automated validation of the change |
|
The PR needs to be labeled with ok-to-test by a maintainer to trigger the automated validation of the change |
Need to check if we can update both g/g and it's api package
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Bumps github.com/gardener/gardener from 1.142.1 to 1.142.6.
Release notes
Sourced from github.com/gardener/gardener's releases.
... (truncated)
Commits
a08f4bdrelease v1.142.630b48a3[release-v1.142] Aligncustomverbauthorizersubject handling with RBAC subj...05d6a62next version: v1.142.6-dev546bfb5release v1.142.57cded6aRespect EXPERIMENTAL_DISABLE_KUBERNETES_VERSION_CHECK flag in seed reconciler...20ab497Update etcd-druid to v0.36.4 (#15014)174d0e2[release-v1.142] Prometheus recording-rule improvements related to metering (...65b1a53next version: v1.142.5-deve6e9818release v1.142.4d171b3f[release-v1.142] Automated cherry pick of #14801: Clean upShootencryption...